Skip to content

feat: custom domains created on Vercel if you include vercel API keys - #27

Merged
carhartlewis merged 5 commits into
mainfrom
lewis/custom-domains
Feb 7, 2025
Merged

carhartlewis merged 5 commits into
mainfrom
lewis/custom-domains

Conversation

@carhartlewis

Copy link
Copy Markdown
Contributor

No description provided.

@vercel

ghost commented Feb 7, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

Name Status Preview Comments Updated (UTC)
app ✅ Ready (Inspect) Visit Preview 💬 Add feedback Feb 7, 2025 5:04pm
1 Skipped Deployment
Name Status Preview Comments Updated (UTC)
web ⬜️ Skipped (Inspect) Feb 7, 2025 5:04pm

@vercel
vercel Bot temporarily deployed to Preview – web February 7, 2025 16:55 Inactive
@vercel
vercel Bot temporarily deployed to Preview – web February 7, 2025 16:58 Inactive

This branch was successfully deployed

1 active and 1 inactive deployments
Preview – app — a813ccd4 Deployed Feb 7, 2025 by vercel[bot]
Preview – web — a813ccd4 Deployed Feb 7, 2025 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants