Skip to content

bump @opennextjs/aws to 3.5.6 - #564

Merged
vicb merged 1 commit into
opennextjs:mainfrom
conico974:chores/bump-aws
Apr 10, 2025
Merged

vicb merged 1 commit into
opennextjs:mainfrom
conico974:chores/bump-aws

Conversation

@conico974

Copy link
Copy Markdown
Collaborator

No description provided.

@changeset-bot

changeset-bot Bot commented Apr 10, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 8f7ed38

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@opennextjs/cloudflare Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@pkg-pr-new

pkg-pr-new Bot commented Apr 10, 2025

Copy link
Copy Markdown

Open in StackBlitz

pnpm add https://pkg.pr.new/@opennextjs/cloudflare@564

commit: 8f7ed38

@vicb vicb left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks!

@vicb
vicb merged commit 6343fb4 into opennextjs:main Apr 10, 2025
@conico974
conico974 deleted the chores/bump-aws branch May 19, 2025 15:25
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants