Skip to content

feat: upload command for worker versions - #559

Merged
james-elicx merged 1 commit into
mainfrom
james/upload-command
Apr 10, 2025
Merged

james-elicx merged 1 commit into
mainfrom
james/upload-command

Conversation

@james-elicx

Copy link
Copy Markdown
Collaborator

There are two ways to create a new version of a worker; immediate deployment with wrangler deploy, or gradual/delayed deployment with wrangler versions upload.

This adds a separate command called opennextjs-cloudflare upload that can be used to upload a new version of a worker.

@changeset-bot

changeset-bot Bot commented Apr 9, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 5bc985f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@opennextjs/cloudflare Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@pkg-pr-new

pkg-pr-new Bot commented Apr 9, 2025

Copy link
Copy Markdown

Open in StackBlitz

pnpm add https://pkg.pr.new/@opennextjs/cloudflare@559

commit: 5bc985f

@james-elicx
james-elicx marked this pull request as ready for review April 9, 2025 21:00
@james-elicx
james-elicx requested a review from vicb April 9, 2025 21:00

@vicb vicb left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM, thanks!

@james-elicx
james-elicx merged commit b002d80 into main Apr 10, 2025
@james-elicx
james-elicx deleted the james/upload-command branch April 10, 2025 11:13
@coderabbitai coderabbitai Bot mentioned this pull request Jul 15, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants