Skip to content

feat(develop-docs): Add telemetry buffer page - #15411

Merged
philippsuess merged 2 commits into
masterfrom
feat/develop-docs-telemtry-buffer
Nov 6, 2025
Merged

philippsuess merged 2 commits into
masterfrom
feat/develop-docs-telemtry-buffer

Conversation

@philippsuess

@philippsuess philippsuess commented Nov 5, 2025 •

Copy link
Copy Markdown
Contributor

DESCRIBE YOUR PR

Add the WIP generic page for telemetry buffers and migrate the batch-processor content.

I can do the more detailed spec for the generic requirements in another PR. This is just to get us started.

IS YOUR CHANGE URGENT?

Help us prioritize incoming PRs by letting us know when the change needs to go live.

  • Urgent deadline (GA date, etc.):
  • Other deadline:
  • None: Not urgent, can wait up to 1 week+

SLA

  • Teamwork makes the dream work, so please add a reviewer to your PRs.
  • Please give the docs team up to 1 week to review your PR unless you've added an urgent due date to it.
    Thanks in advance for your help!

PRE-MERGE CHECKLIST

Make sure you've checked the following before merging your changes:

  • Checked Vercel preview for correctness, including links
  • PR was reviewed and approved by any necessary SMEs (subject matter experts)
  • PR was reviewed and approved by a member of the Sentry docs team

Add the WIP generic page for telemetry buffers and migrate the
batch-processor content.
@vercel

vercel Bot commented Nov 5, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Preview Comments Updated (UTC)
develop-docs Ready Ready Preview Comment Nov 5, 2025 9:19am
1 Skipped Deployment
Project Deployment Preview Comments Updated (UTC)
sentry-docs Ignored Ignored Preview Nov 5, 2025 9:19am

@codecov

codecov Bot commented Nov 5, 2025 •

Copy link
Copy Markdown

Bundle Report

Changes will increase total bundle size by 450 bytes (0.0%) ⬆️. This is within the configured threshold ✅

Detailed changes
Bundle name Size Change
sentry-docs-client-array-push 10.16MB -6 bytes (-0.0%) ⬇️
sentry-docs-server-cjs 12.9MB 456 bytes (0.0%) ⬆️

Affected Assets, Files, and Routes:

view changes for bundle: sentry-docs-client-array-push

Assets Changed:

Asset Name Size Change Total Size Change (%)
static/chunks/pages/_app-*.js -3 bytes 882.71kB -0.0%
static/chunks/8321-*.js -3 bytes 425.87kB -0.0%
server/middleware-*.js -6.46kB 1.0kB -86.59%
server/middleware-*.js 6.46kB 7.46kB 645.5% ⚠️
static/4yAQUUMviWVXiZxwwRwfj/_buildManifest.js (New) 684 bytes 684 bytes 100.0% 🚀
static/4yAQUUMviWVXiZxwwRwfj/_ssgManifest.js (New) 77 bytes 77 bytes 100.0% 🚀
static/16FPsz7i5-*.js (Deleted) -77 bytes 0 bytes -100.0% 🗑️
static/16FPsz7i5-*.js (Deleted) -684 bytes 0 bytes -100.0% 🗑️
view changes for bundle: sentry-docs-server-cjs

Assets Changed:

Asset Name Size Change Total Size Change (%)
1729.js -3 bytes 1.74MB -0.0%
../instrumentation.js -3 bytes 1.07MB -0.0%
9523.js -3 bytes 1.04MB -0.0%
../app/[[...path]]/page.js.nft.json 155 bytes 866.12kB 0.02%
../app/platform-redirect/page.js.nft.json 155 bytes 866.03kB 0.02%
../app/sitemap.xml/route.js.nft.json 155 bytes 863.26kB 0.02%

@Lms24 Lms24 left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This is a good starting point, thanks!

@philippsuess

Copy link
Copy Markdown
Contributor Author

@cleptric, I'm going to merge this. If you have any feedback, I will follow up on it with other PRs.

@philippsuess
philippsuess merged commit 4f189a6 into master Nov 6, 2025
15 checks passed
@philippsuess
philippsuess deleted the feat/develop-docs-telemtry-buffer branch November 6, 2025 06:34
codyde added a commit that referenced this pull request Nov 9, 2025
…ntry/sentry-docs into docs/span-metrics-examples

* 'docs/span-metrics-examples' of https://github.com/getsentry/sentry-docs: (27 commits)
  Adding logs callout in relevant breadcrumbs docs. (#15432)
  docs(self-hosted): Troubleshooting guide to invalidate projectconfigs (#15377)
  docs(js): Add metrics page (#15353)
  fix encoding for minification (#15429)
  update instructions for string minification (#15421)
  getsentry/relay@938582a
  fix: typos in TanStack Start docs (#15425)
  feat: Add firebase docs integration (#15310)
  feat(develop-docs): add index and subpages for Telemetry Buffer (#15424)
  docs(integration): Update Linear Agent Docs  (#15420)
  feat(cloudflare): Fixed snippet to be cloudflare specific (#15369)
  chore(aws-lambda): Add troubleshooting section with known issue (#15414)
  getsentry/relay@5e23f1a
  docs(size-analysis): Align code boxes and remove vcsInfo links (#15418)
  feat(develop-docs): Add telemetry buffer page (#15411)
  Add Composer requirement to PHP platform guide (#15319)
  Updated link for version compatability (#15311)
  (WIP) Mtopo27/size analysis docs (#15256)
  Metrics Product Docs (#15375)
  ref(develop/spans): Replace `is_remote` with `is_segment` in Span Protocol (#15415)
  ...
@github-actions github-actions Bot locked and limited conversation to collaborators Nov 22, 2025

This branch was successfully deployed

2 active (1 outdated) deployments
Preview – develop-docs — 2ab0cc1c Deployed Nov 5, 2025 by vercel[bot]
Preview – sentry-docs — 36d4d77a Deployed Nov 5, 2025 by vercel[bot]
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants