Skip to content

feat(flags): document JS Unleash integration - #12338

Merged
aliu39 merged 6 commits into
masterfrom
aliu/unleash-js
Jan 22, 2025
Merged

aliu39 merged 6 commits into
masterfrom
aliu/unleash-js

Conversation

@aliu39

@aliu39 aliu39 commented Jan 14, 2025 •

Copy link
Copy Markdown
Member

@vercel

vercel Bot commented Jan 14, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

Name Status Preview Comments Updated (UTC)
sentry-docs ✅ Ready (Inspect) Visit Preview 💬 Add feedback Jan 22, 2025 6:29pm
2 Skipped Deployments
Name Status Preview Comments Updated (UTC)
changelog ⬜️ Ignored (Inspect) Visit Preview Jan 22, 2025 6:29pm
develop-docs ⬜️ Ignored (Inspect) Visit Preview Jan 22, 2025 6:29pm

Comment thread docs/platforms/javascript/common/configuration/integrations/unleash.mdx Outdated
Comment thread platform-includes/configuration/unleash/javascript.ember.mdx Outdated
@codecov

codecov Bot commented Jan 14, 2025 •

Copy link
Copy Markdown

Bundle Report

Changes will increase total bundle size by 2.95kB (0.01%) ⬆️. This is within the configured threshold ✅

Detailed changes
Bundle name Size Change
sentry-docs-server-cjs 10.37MB 2.96kB (0.03%) ⬆️
sentry-docs-client-array-push 9.3MB 6 bytes (-0.0%) ⬇️

@aliu39

aliu39 commented Jan 15, 2025

Copy link
Copy Markdown
Member Author

Need to update integration input type in code snippets: see getsentry/sentry-javascript#14948 (comment)

Waiting until the option name is finalized on that PR

@aliu39
aliu39 marked this pull request as draft January 15, 2025 22:00
@michellewzhang

Copy link
Copy Markdown
Contributor

@aliu39 can you ping me when the finalized name is ready? so i can adjust the onboarding snippets (i already merged) if necessary

@aliu39

aliu39 commented Jan 15, 2025

Copy link
Copy Markdown
Member Author

@michellewzhang Ah yeah good callout! Sg

@aliu39
aliu39 enabled auto-merge (squash) January 22, 2025 18:16
@aliu39
aliu39 disabled auto-merge January 22, 2025 18:16
@aliu39
aliu39 enabled auto-merge (squash) January 22, 2025 18:17
@aliu39
aliu39 merged commit c64ddb1 into master Jan 22, 2025
@aliu39
aliu39 deleted the aliu/unleash-js branch January 22, 2025 18:29
@github-actions github-actions Bot locked and limited conversation to collaborators Feb 7, 2025

This branch was successfully deployed

1 active deployment
Preview – sentry-docs — 117b3726 Deployed Jan 22, 2025 by vercel[bot]
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants