Skip to content

fix benchmark script no_build - #2637

Merged
aramikm merged 4 commits into
mainfrom
fix_benchmark_script_no_build
Nov 6, 2025
Merged

aramikm merged 4 commits into
mainfrom
fix_benchmark_script_no_build

Conversation

@aramikm

@aramikm aramikm commented Nov 6, 2025

Copy link
Copy Markdown
Collaborator

Goal

The goal of this PR is to fix the CI issue happening in here

Related to #2601

@aramikm
aramikm requested review from JoeCap08055, enddynayn and shannonwells and removed request for demisx and wilwade November 6, 2025 20:53
Comment thread .github/issues/dependabot-alert.template.md Outdated
Comment thread .github/CODEOWNERS Outdated

@enddynayn enddynayn left a comment

Copy link
Copy Markdown
Collaborator

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

👍

@shannonwells shannonwells left a comment

Copy link
Copy Markdown
Collaborator

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

You can add me to both lists as well

@aramikm
aramikm merged commit 5540b54 into main Nov 6, 2025
33 checks passed
@aramikm
aramikm deleted the fix_benchmark_script_no_build branch November 6, 2025 21:45
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants