Skip to content

[release/6.0] Test on CentOS 7, not RedHat 7 - #58561

Merged
wtgodbe merged 1 commit into
dotnet:release/6.0from
dougbu:dougbu/no.redhat.4298
Nov 5, 2024
Merged

wtgodbe merged 1 commit into
dotnet:release/6.0from
dougbu:dougbu/no.redhat.4298

Conversation

@dougbu

@dougbu dougbu commented Oct 22, 2024

Copy link
Copy Markdown
Contributor

[release/6.0] Test on CentOS 7, not RedHat 7

  • You've read the Contributor Guide and Code of Conduct.
  • You've included unit or integration tests for your change, where applicable.
  • You've included inline docs for your change, where applicable.
  • There's an open issue for the PR that you are making. If you'd like to propose a new feature or change, please open an issue to discuss the change or find an existing issue.

Description

Fixes part of dotnet/dnceng#4298

@ghost ghost added the area-infrastructure Includes: MSBuild projects/targets, build scripts, CI, Installers and shared framework label Oct 22, 2024
@dotnet-policy-service dotnet-policy-service Bot added this to the 6.0.x milestone Oct 22, 2024
- dotnet/dnceng#4298 part II/II
  - contributes to dotnet/dnceng#3189
- use Docker container for CentOS, mimicking dotnet/runtime#105063
@dougbu dougbu added the tell-mode Indicates a PR which is being merged during tell-mode label Oct 22, 2024
@dougbu
dougbu requested a review from a team October 23, 2024 00:52
@dougbu
dougbu marked this pull request as ready for review October 23, 2024 00:52
@dougbu

dougbu commented Oct 23, 2024

Copy link
Copy Markdown
Contributor Author

hi @halter73, @BrennanConroy, @wtgodbe, and other @dotnet/aspnet-build members. I'm just visiting 😁

oh @Tratcher is still in the CODEOWNERS file. nice and hi to you too

@dougbu

dougbu commented Oct 23, 2024

Copy link
Copy Markdown
Contributor Author

/ping

@wtgodbe wtgodbe left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM, but are we planning on shipping another 6.0 release?

@dougbu

dougbu commented Oct 28, 2024

Copy link
Copy Markdown
Contributor Author

on hold until release/6.0 branch opens again

@dotnet-policy-service dotnet-policy-service Bot added the pending-ci-rerun When assigned to a PR indicates that the CI checks should be rerun label Nov 5, 2024
@dougbu

dougbu commented Nov 5, 2024

Copy link
Copy Markdown
Contributor Author

/azp run

@azure-pipelines

Copy link
Copy Markdown
Azure Pipelines successfully started running 2 pipeline(s).

@dougbu

dougbu commented Nov 5, 2024

Copy link
Copy Markdown
Contributor Author

/ping @wtgodbe for final squish 'n merge once build goes through 😁

@dotnet-policy-service dotnet-policy-service Bot removed the pending-ci-rerun When assigned to a PR indicates that the CI checks should be rerun label Nov 5, 2024
@wtgodbe
wtgodbe merged commit e185938 into dotnet:release/6.0 Nov 5, 2024
@dotnet-policy-service dotnet-policy-service Bot modified the milestones: 6.0.x, 6.0.36 Nov 5, 2024
@dougbu
dougbu deleted the dougbu/no.redhat.4298 branch November 5, 2024 21:28
@dougbu

dougbu commented Nov 5, 2024

Copy link
Copy Markdown
Contributor Author

thanx @wtgodbe❗

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

area-infrastructure Includes: MSBuild projects/targets, build scripts, CI, Installers and shared framework tell-mode Indicates a PR which is being merged during tell-mode

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants