Skip to content

chore: update CODEOWNERS - #23

Merged
louis-bompart merged 1 commit into
mainfrom
update-codeowners-dx
Jan 11, 2023
Merged

louis-bompart merged 1 commit into
mainfrom
update-codeowners-dx

Conversation

@louis-bompart

Copy link
Copy Markdown
Contributor

@louis-bompart
louis-bompart merged commit 16229dd into main Jan 11, 2023
@y-lakhdar
y-lakhdar deleted the update-codeowners-dx branch June 15, 2023 18:21
y-lakhdar pushed a commit that referenced this pull request Jun 20, 2023
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants