Skip to content

[CI:DOCS] point IRC to libera.chat - #10499

Merged
openshift-merge-robot merged 1 commit into
podman-container-tools:masterfrom
edsantiago:libera
May 29, 2021
Merged

openshift-merge-robot merged 1 commit into
podman-container-tools:masterfrom
edsantiago:libera

Conversation

@edsantiago

Copy link
Copy Markdown
Contributor

Project IRC comms are moving to Libera. Update docs.

Confirmation cookie: libera-ieVeeGahbiaf1einguw1xav6bahquie

Signed-off-by: Ed Santiago santiago@redhat.com

Project IRC comms are moving to Libera. Update docs.

Confirmation cookie: libera-ieVeeGahbiaf1einguw1xav6bahquie

Signed-off-by: Ed Santiago <santiago@redhat.com>
@openshift-ci

openshift-ci Bot commented May 29, 2021

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: edsantiago

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 29, 2021
@edsantiago

Copy link
Copy Markdown
Contributor Author

Cookie as requested by Libera.Chat ticket number 421, channel registration.

@edsantiago

Copy link
Copy Markdown
Contributor Author

@containers/podman-maintainers in case anyone is around this weekend: merging this will grant us ownership of the #podman namespace on libera.chat

@Luap99 Luap99 left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM

@Luap99

Luap99 commented May 29, 2021

Copy link
Copy Markdown
Member

I think podman.io also needs to be updated.

@mheon

mheon commented May 29, 2021

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label May 29, 2021
@openshift-merge-robot
openshift-merge-robot merged commit ffca97a into podman-container-tools:master May 29, 2021
@edsantiago
edsantiago deleted the libera branch May 29, 2021 23:19
@github-actions github-actions Bot added the locked - please file new issue/PR Assist humans wanting to comment on an old issue or PR with locked comments. label Sep 22, 2023
@github-actions github-actions Bot locked as resolved and limited conversation to collaborators Sep 22, 2023
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. locked - please file new issue/PR Assist humans wanting to comment on an old issue or PR with locked comments.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants