Skip to content

fix(clerk-js): Do not bubble up CaptchaHeartbeat errors - #5032

Merged
LekoArts merged 8 commits into
mainfrom
fix/dont-bubble-up-captcha-errors
Jan 29, 2025
Merged

LekoArts merged 8 commits into
mainfrom
fix/dont-bubble-up-captcha-errors

Conversation

@jacekradko

@jacekradko jacekradko commented Jan 29, 2025 •

Copy link
Copy Markdown
Contributor

Description

Prevent unsafe optional chaining error in CaptchaHeartbeat from bubbling up to the host application. Need to look into larger refactoring of non-null assertions in the codebase.

Fixes: #5027

Checklist

  • pnpm test runs as expected.
  • pnpm build runs as expected.
  • (If applicable) JSDoc comments have been added or updated for any package exports
  • (If applicable) Documentation has been updated

Type of change

  • 🐛 Bug fix
  • 🌟 New feature
  • 🔨 Breaking change
  • 📖 Refactoring / dependency upgrade / documentation
  • other:

@vercel

vercel Bot commented Jan 29, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

Name Status Preview Comments Updated (UTC)
clerk-js-sandbox ✅ Ready (Inspect) Visit Preview 💬 Add feedback Jan 29, 2025 4:48pm

@changeset-bot

changeset-bot Bot commented Jan 29, 2025 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: c48fead

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 3 packages
Name Type
@clerk/clerk-js Patch
@clerk/chrome-extension Patch
@clerk/clerk-expo Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

Comment thread packages/clerk-js/src/core/auth/CaptchaHeartbeat.ts Outdated

@brkalow brkalow left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

👏

@LekoArts
LekoArts enabled auto-merge (squash) January 29, 2025 09:34

@anagstef anagstef left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

The changeset is kind of misleading

Comment thread .changeset/stale-badgers-scream.md Outdated
Co-authored-by: Stefanos Anagnostou <anagstef@users.noreply.github.com>

This branch was successfully deployed

1 active deployment
Preview — c48fead9 Deployed Jan 29, 2025 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

bug Something isn't working clerk-js

Projects

None yet

Development

Successfully merging this pull request may close these issues.

TypeError: Cannot read properties of undefined (reading 'captchaHeartbeat')

6 participants