Skip to content

Update CODEOWNERS - #4

Merged
RomaniukNatalia merged 1 commit into
mainfrom
RomaniukNatalia-patch-1
Dec 2, 2024
Merged

RomaniukNatalia merged 1 commit into
mainfrom
RomaniukNatalia-patch-1

Conversation

@RomaniukNatalia

Copy link
Copy Markdown
Owner

No description provided.

Signed-off-by: Nataliia_Rmnk <57407326+RomaniukNatalia@users.noreply.github.com>
@RomaniukNatalia
RomaniukNatalia merged commit cdb99df into main Dec 2, 2024
@RomaniukNatalia
RomaniukNatalia deleted the RomaniukNatalia-patch-1 branch December 2, 2024 14:22
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant