fix: Add support for multiple APIPaths in caddygateway - #1137
Open
ezr-ondrej wants to merge 5 commits into
Open
ezr-ondrej wants to merge 5 commits into
ezr-ondrej wants to merge 5 commits into
Conversation
Collaborator
|
Good fix - is it possible to get a test in here too? |
ezr-ondrej
force-pushed
the
caddy_api_paths
branch
from
March 19, 2025 12:27
93f0006 to
d749094
Compare
Member
Author
|
I've refactored the method a bit to separate the logic for preparation of the paths. |
Collaborator
|
@wcmitchell a good one to review ;) |
Refactor the relevant part of the code away from the main function. That helps mocking the Applications for which we want to build paths. This allows us to cover building the paths for the gateway by separate tests.
ezr-ondrej
force-pushed
the
caddy_api_paths
branch
from
March 25, 2025 21:30
d749094 to
66df59b
Compare
Contributor
|
/retest |
Contributor
|
@ezr-ondrej I'm seeing a Kuttl test failure for |
This branch has not been deployed
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
No description provided.