Skip to content
This repository was archived by the owner on Oct 7, 2024. It is now read-only.

chore(deps): bump @metamask/abi-utils from 2.0.2 to 2.0.4 - #405

Open
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/main/metamask/abi-utils-2.0.4
Open

dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/main/metamask/abi-utils-2.0.4

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Jul 10, 2024

Copy link
Copy Markdown
Contributor

Bumps @metamask/abi-utils from 2.0.2 to 2.0.4.

Release notes

Sourced from @​metamask/abi-utils's releases.

2.0.4

Changed

  • Bump @metamask/superstruct from ^3.0.0 to ^3.1.0 (#79, #80)
  • Bump @metamask/utils from ^8.5.0 to ^9.0.0 (#80)

2.0.3

Changed

  • Bump @metamask/utils from ^8.1.0 to ^8.5.0 (#73, #67, #64, #58)

Fixed

  • Replace dependency superstruct with @metamask/superstruct ^3.0.0 (#73)
    • This fixes the issue of this package being unusable by any TypeScript project that uses Node16 or NodeNext as its moduleResolution option.
Changelog

Sourced from @​metamask/abi-utils's changelog.

[2.0.4]

Changed

  • Bump @metamask/superstruct from ^3.0.0 to ^3.1.0 (#79, #80)
  • Bump @metamask/utils from ^8.5.0 to ^9.0.0 (#80)

[2.0.3]

Changed

  • Bump @metamask/utils from ^8.1.0 to ^8.5.0 (#73, #67, #64, #58)

Fixed

  • Replace dependency superstruct with @metamask/superstruct ^3.0.0 (#73)
    • This fixes the issue of this package being unusable by any TypeScript project that uses Node16 or NodeNext as its moduleResolution option.
Commits
  • 5a897fe 2.0.4 (#81)
  • 6d8738e Bump @metamask/superstruct to ^3.1.0, @metamask/utils to ^9.0.0 (#80)
  • d128a53 Bump @​metamask/superstruct from 3.0.0 to 3.1.0 (#79)
  • 2480526 2.0.3 (#78)
  • b5d6e8b Bump the npm_and_yarn group across 1 directory with 2 updates (#76)
  • fe34d37 Bump tar in the npm_and_yarn group across 1 directory (#74)
  • 7179f32 Replace 'superstruct' imports with '@​metamask/superstruct' (#73)
  • 7cde6dd Update CODEOWNERS (#71)
  • b4f929d Enabling MetaMask security code scanner (#68)
  • 5aad2c5 Add Shared Libraries team in codeowners file (#66)
  • Additional commits viewable in compare view

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot merge will merge this PR after your CI passes on it
  • @dependabot squash and merge will squash and merge this PR after your CI passes on it
  • @dependabot cancel merge will cancel a previously requested merge and block automerging
  • @dependabot reopen will reopen this PR if it is closed
  • @dependabot close will close this PR and stop Dependabot recreating it. You can achieve the same result by closing it manually
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [@metamask/abi-utils](https://github.com/MetaMask/abi-utils) from 2.0.2 to 2.0.4.
- [Release notes](https://github.com/MetaMask/abi-utils/releases)
- [Changelog](https://github.com/MetaMask/abi-utils/blob/main/CHANGELOG.md)
- [Commits](MetaMask/abi-utils@v2.0.2...v2.0.4)

---
updated-dependencies:
- dependency-name: "@metamask/abi-utils"
  dependency-type: indirect
  update-type: version-update:semver-patch
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabot Bot requested a review from a team as a code owner July 10, 2024 06:04
@dependabot dependabot Bot added the type-dependencies Pull requests that update a dependency file label Jul 10, 2024

This branch has not been deployed

No deployments
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

type-dependencies Pull requests that update a dependency file

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants