Skip to content

[PAY-3156] Fix Rewards Claiming Analytics - #9024

Merged
rickyrombo merged 14 commits into
mainfrom
mjp-challenge-analytics
Jul 8, 2024
Merged

[PAY-3156] Fix Rewards Claiming Analytics#9024
rickyrombo merged 14 commits into
mainfrom
mjp-challenge-analytics

Conversation

@rickyrombo

@rickyrombo rickyrombo commented Jul 2, 2024

Copy link
Copy Markdown
Contributor

Description

PAY-3156
Changes the analytics for Rewards and adds support for them in the SDK flow. Removes non-SDK flow.

How Has This Been Tested?

Tested locally

@changeset-bot

ghost commented Jul 2, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 745e7e1

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 2 packages
Name Type
@audius/sdk Major
@audius/sp-actions Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@rickyrombo rickyrombo changed the title Fix Rewards Claiming Analytics [PAY-3156] Fix Rewards Claiming Analytics Jul 2, 2024
@audius-infra

ghost commented Jul 2, 2024

Copy link
Copy Markdown
Collaborator

Preview this change https://demo.audius.co/mjp-challenge-analytics

Comment thread .changeset/soft-kiwis-swim.md
Comment thread packages/libs/src/sdk/utils/errors.ts
const isResponseError = (error: Error): error is ResponseError =>
'response' in error && error.response instanceof Response
export const isResponseError = (error: unknown): error is ResponseError =>
error instanceof Error && error.name === 'ResponseError'

ghost Jul 3, 2024

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Responded about this in Slack, but I'm wondering if we shouldn't just update our compilation target? instanceof is a pretty useful/common pattern for checking things and since we don't have access to the internals of third-party code, it would be nice to maintain the ability to use it.

@rickyrombo
rickyrombo enabled auto-merge (squash) July 8, 2024 21:43
@audius-infra

ghost commented Jul 8, 2024

Copy link
Copy Markdown
Collaborator

Preview this change https://demo.audius.co/mjp-challenge-analytics

@rickyrombo
rickyrombo merged commit 92d4d81 into main Jul 8, 2024
@rickyrombo
rickyrombo deleted the mjp-challenge-analytics branch July 8, 2024 22:26
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants